Skip to content

Support Storybook 10 - #62

Merged
ljcl merged 1 commit into
mainfrom
storybook-10-support
Nov 17, 2025
Merged

Support Storybook 10#62
ljcl merged 1 commit into
mainfrom
storybook-10-support

Conversation

@ljcl

@ljcl ljcl commented Nov 17, 2025

Copy link
Copy Markdown
Owner

@changeset-bot

changeset-bot Bot commented Nov 17, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: cf63647

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@ljcl/storybook-addon-cssprops Major

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@ljcl
ljcl merged commit 2b690f0 into main Nov 17, 2025
1 check passed
@ljcl
ljcl deleted the storybook-10-support branch November 17, 2025 21:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Storybook 10 support

1 participant