Skip to content

support CSF Factories - #64

Merged
ljcl merged 1 commit into
mainfrom
refactor-builder
Dec 16, 2025
Merged

support CSF Factories#64
ljcl merged 1 commit into
mainfrom
refactor-builder

Conversation

@ljcl

@ljcl ljcl commented Dec 16, 2025

Copy link
Copy Markdown
Owner

No description provided.

@changeset-bot

changeset-bot Bot commented Dec 16, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c9bb0a6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@ljcl/storybook-addon-cssprops Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@ljcl
ljcl merged commit e93be8a into main Dec 16, 2025
1 check passed
@ljcl
ljcl deleted the refactor-builder branch December 16, 2025 03:22
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant